Skip to content

fix(burner): remove burner creation in react sdk due to starknetjs issue - #346

Merged
kwiss merged 2 commits into
mainfrom
fix/react-burner
Apr 26, 2024
Merged

kwiss merged 2 commits into
mainfrom
fix/react-burner

Conversation

@kwiss

@kwiss kwiss commented Apr 26, 2024

Copy link
Copy Markdown
Contributor

Description

What type of PR is this? (check all applicable)

  • 🍕 Feature (feat:)
  • 🐛 Bug Fix (fix:)
  • 📝 Documentation Update (docs:)
  • 🎨 Style (style:)
  • 🧑‍💻 Code Refactor (refactor:)
  • 🔥 Performance Improvements (perf:)
  • ✅ Test (test:)
  • 🤖 Build (build:)
  • 🔁 CI (ci:)
  • 📦 Chore (chore:)
  • ⏩ Revert (revert:)
  • 🚀 Breaking Changes (BREAKING CHANGE:)

Related Tickets & Documents

Added tests?

  • 👍 yes
  • 🙅 no, because they aren't needed
  • 🙋 no, because I need help

Added to documentation?

  • 📜 README.md
  • 📓 Documentation
  • 🙅 no documentation needed

[optional] Are there any post-deployment tasks we need to perform?

[optional] What gif best describes this PR or how it makes you feel?

PR Title and Description Guidelines:

  • Ensure your PR title follows semantic versioning standards. This helps automate releases and changelogs.
  • Use types like feat:, fix:, chore:, BREAKING CHANGE: etc. in your PR title.
  • Your PR title will be used as a commit message when merging. Make sure it adheres to Conventional Commits standards.

Closing Issues

@changeset-bot

changeset-bot Bot commented Apr 26, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b1a86df

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 3 packages
Name Type
@ark-project/react Minor
@ark-project/core Patch
@ark-project/demo Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@vercel

vercel Bot commented Apr 26, 2024

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

Name Status Preview Comments Updated (UTC)
ark-project-sdk-demo ✅ Ready (Inspect) Visit Preview 💬 Add feedback Apr 26, 2024 5:44pm

@kwiss
kwiss marked this pull request as ready for review April 26, 2024 17:50
@kwiss
kwiss merged commit 5dcd498 into main Apr 26, 2024
@kwiss
kwiss deleted the fix/react-burner branch April 26, 2024 17:51

This branch was successfully deployed

1 active deployment
Preview b1a86df4 Deployed Apr 26, 2024 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant