Skip to content

fix(sdk): fix react sdk cancel not using useArkAccount - #348

Merged
kwiss merged 2 commits into
mainfrom
fix/react-sdk-cancel
Apr 26, 2024
Merged

kwiss merged 2 commits into
mainfrom
fix/react-sdk-cancel

Conversation

@kwiss

@kwiss kwiss commented Apr 26, 2024

Copy link
Copy Markdown
Contributor

Description

Fix react SDK, update useCancel to use useBurner instead of directly declaring a new account

What type of PR is this? (check all applicable)

  • 🐛 Bug Fix (fix:)

Related Tickets & Documents

Added tests?

  • 👍 yes
  • 🙅 no, because they aren't needed
  • 🙋 no, because I need help

Added to documentation?

  • 📜 README.md
  • 📓 Documentation
  • 🙅 no documentation needed

[optional] Are there any post-deployment tasks we need to perform?

[optional] What gif best describes this PR or how it makes you feel?

PR Title and Description Guidelines:

  • Ensure your PR title follows semantic versioning standards. This helps automate releases and changelogs.
  • Use types like feat:, fix:, chore:, BREAKING CHANGE: etc. in your PR title.
  • Your PR title will be used as a commit message when merging. Make sure it adheres to Conventional Commits standards.

Closing Issues

@changeset-bot

changeset-bot Bot commented Apr 26, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: e86a1c9

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 2 packages
Name Type
@ark-project/react Minor
@ark-project/demo Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@vercel

vercel Bot commented Apr 26, 2024

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

Name Status Preview Comments Updated (UTC)
ark-project-sdk-demo ✅ Ready (Inspect) Visit Preview 💬 Add feedback Apr 26, 2024 9:01pm

@kwiss
kwiss marked this pull request as ready for review April 26, 2024 20:56
@kwiss
kwiss merged commit 90fa414 into main Apr 26, 2024
@kwiss
kwiss deleted the fix/react-sdk-cancel branch April 26, 2024 21:13

This branch was successfully deployed

1 active deployment
Preview e86a1c9d Deployed Apr 26, 2024 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant