Skip to content

feat(deployer): update for sepolia - #394

Merged
kwiss merged 4 commits into
mainfrom
feat/deployer-sepolia
Jun 17, 2024
Merged

kwiss merged 4 commits into
mainfrom
feat/deployer-sepolia

Conversation

@kwiss

@kwiss kwiss commented Jun 16, 2024

Copy link
Copy Markdown
Contributor

Description

What type of PR is this? (check all applicable)

  • 🍕 Feature (feat:)
  • 🐛 Bug Fix (fix:)
  • 📝 Documentation Update (docs:)
  • 🎨 Style (style:)
  • 🧑‍💻 Code Refactor (refactor:)
  • 🔥 Performance Improvements (perf:)
  • ✅ Test (test:)
  • 🤖 Build (build:)
  • 🔁 CI (ci:)
  • 📦 Chore (chore:)
  • ⏩ Revert (revert:)
  • 🚀 Breaking Changes (BREAKING CHANGE:)

Related Tickets & Documents

Added tests?

  • 👍 yes
  • 🙅 no, because they aren't needed
  • 🙋 no, because I need help

Added to documentation?

  • 📜 README.md
  • 📓 Documentation
  • 🙅 no documentation needed

[optional] Are there any post-deployment tasks we need to perform?

[optional] What gif best describes this PR or how it makes you feel?

PR Title and Description Guidelines:

  • Ensure your PR title follows semantic versioning standards. This helps automate releases and changelogs.
  • Use types like feat:, fix:, chore:, BREAKING CHANGE: etc. in your PR title.
  • Your PR title will be used as a commit message when merging. Make sure it adheres to Conventional Commits standards.

Closing Issues

@vercel

vercel Bot commented Jun 16, 2024

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

Name Status Preview Comments Updated (UTC)
ark-project-sdk-demo ✅ Ready (Inspect) Visit Preview 💬 Add feedback Jun 17, 2024 11:30am

@changeset-bot

changeset-bot Bot commented Jun 16, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 467d2c7

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 4 packages
Name Type
@ark-project/core Patch
@ark-project/react Patch
@ark-project/demo Patch
@ark-project/core-examples Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@kwiss
kwiss marked this pull request as ready for review June 17, 2024 11:26
@kwiss
kwiss merged commit e0306a2 into main Jun 17, 2024
@kwiss
kwiss deleted the feat/deployer-sepolia branch June 17, 2024 11:32
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants