Skip to content

Chore/bump sn react - #519

Merged
kwiss merged 1 commit into
ArkProjectNFTs:feat/contract-v2from
MartianGreed:chore/bump-sn-react
Mar 17, 2025
Merged

kwiss merged 1 commit into
ArkProjectNFTs:feat/contract-v2from
MartianGreed:chore/bump-sn-react

Conversation

@MartianGreed

Copy link
Copy Markdown

Description

What type of PR is this? (check all applicable)

  • 🍕 Feature (feat:)
  • 🐛 Bug Fix (fix:)
  • 📝 Documentation Update (docs:)
  • 🎨 Style (style:)
  • 🧑‍💻 Code Refactor (refactor:)
  • 🔥 Performance Improvements (perf:)
  • ✅ Test (test:)
  • 🤖 Build (build:)
  • 🔁 CI (ci:)
  • 📦 Chore (chore:)
  • ⏩ Revert (revert:)
  • 🚀 Breaking Changes (BREAKING CHANGE:)

Related Tickets & Documents

Added tests?

  • 👍 yes
  • 🙅 no, because they aren't needed
  • 🙋 no, because I need help

Added to documentation?

  • 📜 README.md
  • 📓 Documentation
  • 🙅 no documentation needed

[optional] Are there any post-deployment tasks we need to perform?

[optional] What gif best describes this PR or how it makes you feel?

PR Title and Description Guidelines:

  • Ensure your PR title follows semantic versioning standards. This helps automate releases and changelogs.
  • Use types like feat:, fix:, chore:, BREAKING CHANGE: etc. in your PR title.
  • Your PR title will be used as a commit message when merging. Make sure it adheres to Conventional Commits standards.

Closing Issues

@vercel

vercel Bot commented Mar 17, 2025

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

Name Status Preview Comments Updated (UTC)
ark-book ❌ Failed (Inspect) Mar 17, 2025 2:33pm

@changeset-bot

changeset-bot Bot commented Mar 17, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 74ceb6b

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 2 packages
Name Type
@ark-project/react Patch
@ark-project/core Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@MartianGreed
MartianGreed changed the base branch from main to feat/contract-v2 March 17, 2025 14:33
@kwiss
kwiss merged commit a0f6ce0 into ArkProjectNFTs:feat/contract-v2 Mar 17, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants