Skip to content

feat: compute_reservation_create_template - #3811

Merged
subfuzion merged 8 commits into
mainfrom
compute_reservation_create_template
Sep 11, 2024
Merged

subfuzion merged 8 commits into
mainfrom
compute_reservation_create_template

Conversation

@gryczj

@gryczj gryczj commented Aug 28, 2024 •

Copy link
Copy Markdown
Contributor

Description

Fixes #

Note: Before submitting a pull request, please open an issue for discussion if you are not associated with Google.

Checklist

  • I have followed guidelines from CONTRIBUTING.MD and Samples Style Guide
  • Tests pass: npm test (see Testing)
  • Lint pass: npm run lint (see Style)
  • These samples need a new API enabled in testing projects to pass (let us know which ones)
  • These samples need a new/updated env vars in testing projects set to pass (let us know which ones)
  • This pull request is from a branch created directly off of GoogleCloudPlatform/nodejs-docs-samples. Not a fork.
  • This sample adds a new sample directory, and I updated the CODEOWNERS file with the codeowners for this sample
  • This sample adds a new sample directory, and I created GitHub Actions workflow for this sample
  • This sample adds a new Product API, and I updated the Blunderbuss issue/PR auto-assigner with the codeowners for this sample
  • Please merge this PR for me once it is approved

@gryczj gryczj self-assigned this Aug 28, 2024
@product-auto-label product-auto-label Bot added samples Issues that are directly related to samples. api: compute Issues related to the Compute Engine API. labels Aug 28, 2024
@gryczj gryczj added kokoro:force-run Add this label to force Kokoro to re-run the tests. kokoro:run Add this label to force Kokoro to re-run the tests. labels Aug 28, 2024
@gryczj
gryczj force-pushed the compute_reservation_create_template branch 2 times, most recently from 57542b0 to 91048d1 Compare August 28, 2024 16:40
@kokoro-team kokoro-team removed kokoro:run Add this label to force Kokoro to re-run the tests. kokoro:force-run Add this label to force Kokoro to re-run the tests. labels Aug 28, 2024
@gryczj
gryczj force-pushed the compute_reservation_create_template branch 9 times, most recently from 8cf6ee0 to 9bb5e8c Compare September 2, 2024 16:47
@gryczj
gryczj marked this pull request as ready for review September 2, 2024 16:47
@gryczj
gryczj requested review from a team as code owners September 2, 2024 16:47
@snippet-bot

snippet-bot Bot commented Sep 2, 2024 •

Copy link
Copy Markdown

Here is the summary of changes.

You are about to add 1 region tag.

This comment is generated by snippet-bot.
If you find problems with this result, please file an issue at:
https://github.com/googleapis/repo-automation-bots/issues.
To update this comment, add snippet-bot:force-run label or use the checkbox below:

  • Refresh this comment

@subfuzion

Copy link
Copy Markdown
Contributor

Pending #3807

@subfuzion
subfuzion self-requested a review September 3, 2024 18:50
@gryczj
gryczj force-pushed the compute_reservation_create_template branch 2 times, most recently from 43ee23d to a0479dc Compare September 4, 2024 15:02
@gryczj

gryczj commented Sep 6, 2024

Copy link
Copy Markdown
Contributor Author

Please review #3834 first

@gryczj
gryczj force-pushed the compute_reservation_create_template branch from a0479dc to 263dcbb Compare September 6, 2024 13:48

@subfuzion subfuzion left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

👍

@subfuzion
subfuzion merged commit b0f95bc into main Sep 11, 2024
@subfuzion
subfuzion deleted the compute_reservation_create_template branch September 11, 2024 22:25
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

api: compute Issues related to the Compute Engine API. samples Issues that are directly related to samples.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants