Skip to content

chore(config): migrate Renovate config - autoclosed - #872

Merged
jacobk999 merged 2 commits into
mainfrom
renovate/migrate-config
Jun 3, 2026
Merged

chore(config): migrate Renovate config - autoclosed#872
jacobk999 merged 2 commits into
mainfrom
renovate/migrate-config

Conversation

@renovate

@renovate renovate Bot commented Jun 2, 2026

Copy link
Copy Markdown
Contributor

The Renovate config in this repository needs migrating. Typically this is because one or more configuration options you are using have been renamed.

You don't need to merge this PR right away, because Renovate will continue to migrate these fields internally each time it runs. But later some of these fields may be fully deprecated and the migrations removed. So it's a good idea to merge this migration PR soon.

🔕 Ignore: Close this PR and you won't be reminded about config migration again, but one day your current config may no longer be valid.

❓ Got questions? Does something look wrong to you? Please don't hesitate to request help here.


This PR was generated by Mend Renovate. View the repository job log.

@renovate renovate Bot added the dependencies label Jun 2, 2026
@vercel

vercel Bot commented Jun 2, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated (UTC)
statsify Ignored Ignored Preview Jun 3, 2026 12:59am

@renovate renovate Bot mentioned this pull request Jun 2, 2026
24 tasks
@jacobk999
jacobk999 merged commit e81b2f9 into main Jun 3, 2026
9 checks passed
@jacobk999
jacobk999 deleted the renovate/migrate-config branch June 3, 2026 01:00
@renovate renovate Bot changed the title chore(config): migrate Renovate config chore(config): migrate Renovate config - autoclosed Jun 3, 2026
jacobk999 added a commit that referenced this pull request Jun 8, 2026
* chore(config): migrate config renovate.json

* chore: let all dependencies be updatable

---------

Co-authored-by: renovate[bot] <29139614+renovate[bot]@users.noreply.github.com>
Co-authored-by: jacobk999 <jababakoshy@gmail.com>
ugcodrr pushed a commit that referenced this pull request Jun 16, 2026
* chore(config): migrate config renovate.json

* chore: let all dependencies be updatable

---------

Co-authored-by: renovate[bot] <29139614+renovate[bot]@users.noreply.github.com>
Co-authored-by: jacobk999 <jababakoshy@gmail.com>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant