Skip to content

fix(font-renderer): Add back &r (reset) formatting behavior - #891

Merged
jacobk999 merged 1 commit into
mainfrom
fix/leaderboard-bold
Jun 12, 2026
Merged

fix(font-renderer): Add back &r (reset) formatting behavior#891
jacobk999 merged 1 commit into
mainfrom
fix/leaderboard-bold

Conversation

@ugcodrr

@ugcodrr ugcodrr commented Jun 11, 2026

Copy link
Copy Markdown
Member

Fixes bold leaderboards, and the &r reset colorcode functionality, also includes a test for FontRenderer lex method to ensure formatting resets correctly

* use Object.assign to avoid state mutation
* add test for FontRenderer lex method to ensure formatting resets correctly
@vercel

vercel Bot commented Jun 11, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated (UTC)
statsify Ignored Ignored Jun 11, 2026 5:02am

@jacobk999
jacobk999 merged commit 406024e into main Jun 12, 2026
9 checks passed
@jacobk999
jacobk999 deleted the fix/leaderboard-bold branch June 12, 2026 00:10
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants