Skip to content

chore: update @types/react and @types/react-dom - #1859

Open
rssilva wants to merge 3 commits into
mainfrom
chore-update-react-related-types
Open

rssilva wants to merge 3 commits into
mainfrom
chore-update-react-related-types

Conversation

@rssilva

@rssilva rssilva commented May 7, 2025

Copy link
Copy Markdown
Contributor

Description of the change

Update @types/react and @types/react-dom to 19

Type of change

  • Non-Breaking Change (change to existing functionality)

@rssilva rssilva self-assigned this May 7, 2025
@rssilva
rssilva force-pushed the chore-update-react-related-types branch from 06c5e61 to ed1fdd9 Compare May 7, 2025 13:12
@changeset-bot

changeset-bot Bot commented May 7, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ed1fdd9

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 2 packages
Name Type
@localyze-pluto/components Major
@localyze-pluto/theme Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants