Skip to content

bug: support customized spore script for transfer&melt method - #175

Merged
Hanssen0 merged 1 commit into
ckb-devrel:masterfrom
ashuralyk:bug/spore-script-info
Mar 20, 2025
Merged

Hanssen0 merged 1 commit into
ckb-devrel:masterfrom
ashuralyk:bug/spore-script-info

Conversation

@ashuralyk

Copy link
Copy Markdown
Contributor

Description

Reported by UTXOGlobal team that methods transferSpore and meltSpore are not allowed for user-defined spore scripts. To solve this issue, we should expose ScriptInfoLike parameter.

@changeset-bot

changeset-bot Bot commented Mar 12, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 96b0e81

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 8 packages
Name Type
@ckb-ccc/spore Patch
@ckb-ccc/shell Patch
@ckb-ccc/ccc Patch
ckb-ccc Patch
@ckb-ccc/connector Patch
@ckb-ccc/examples Patch
@ckb-ccc/ccc-playground Patch
@ckb-ccc/connector-react Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

Comment thread packages/spore/src/cluster/index.ts Outdated
id: ccc.HexLike;
to: ccc.ScriptLike;
tx?: ccc.TransactionLike;
scriptInfo?: SporeScriptInfoLike;

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Can we use the same args as assertCluster - scripts to allow auto choosing scriptInfo?

@ashuralyk
ashuralyk force-pushed the bug/spore-script-info branch from c83bd7c to 522d663 Compare March 20, 2025 08:01
@ashuralyk
ashuralyk requested a review from Hanssen0 March 20, 2025 08:01
@ashuralyk
ashuralyk force-pushed the bug/spore-script-info branch from 522d663 to 96b0e81 Compare March 20, 2025 08:57

@Hanssen0 Hanssen0 left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants