Skip to content

feat(core): SignerNostrPublicKeyReadonly - #182

Merged
Hanssen0 merged 2 commits into
ckb-devrel:masterfrom
Hanssen0:master
Mar 21, 2025
Merged

Hanssen0 merged 2 commits into
ckb-devrel:masterfrom
Hanssen0:master

Conversation

@Hanssen0

Copy link
Copy Markdown
Member

No description provided.

@changeset-bot

changeset-bot Bot commented Mar 21, 2025 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b210e51

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 20 packages
Name Type
@ckb-ccc/core Minor
@ckb-ccc/eip6963 Patch
@ckb-ccc/joy-id Patch
@ckb-ccc/lumos-patches Patch
@ckb-ccc/nip07 Patch
@ckb-ccc/okx Patch
@ckb-ccc/rei Patch
@ckb-ccc/shell Patch
@ckb-ccc/spore Patch
@ckb-ccc/ssri Patch
@ckb-ccc/udt Patch
@ckb-ccc/uni-sat Patch
@ckb-ccc/utxo-global Patch
@ckb-ccc/xverse Patch
@ckb-ccc/ccc Patch
ckb-ccc Patch
@ckb-ccc/connector Patch
@ckb-ccc/examples Patch
@ckb-ccc/ccc-playground Patch
@ckb-ccc/connector-react Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@Hanssen0
Hanssen0 force-pushed the master branch 3 times, most recently from 8bb7ff9 to 3d69fc7 Compare March 21, 2025 14:15

@yixyxiu yixyxiu left a comment

Copy link
Copy Markdown
Collaborator

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants