Skip to content

Add c-card--contained modifier - #1287

Merged
tylersticka merged 2 commits into
v-nextfrom
feature/contained-cards-r2
Jun 7, 2021
Merged

Add c-card--contained modifier#1287
tylersticka merged 2 commits into
v-nextfrom
feature/contained-cards-r2

Conversation

@tylersticka

@tylersticka tylersticka commented Jun 7, 2021

Copy link
Copy Markdown
Member

Overview

  • Adds c-card--contained modifier, giving cards distinct boundaries to offset them from their containing element.
  • Adds an unthemed-styles mixin to define styles that should apply only when a theme class is not directly attached to the element.

This PR does not include the distinctly separated footer from the rest of the card contents. I realized partway into this task that making this happen would represent 95% of the effort, but that design detail is only present on PWA Stats. Because there's another issue blocked by this and many more directly related to our short-term goals, it felt odd to spend many more hours (if not days) refactoring for this design detail we wouldn't use right away. I plan to create another issue to address that detail at a later date.

Screenshots

Screenshot 2021-06-07 at 14-26-36 Storybook

Screenshot 2021-06-07 at 14-25-56 Storybook

Testing

Deploy preview


@changeset-bot

changeset-bot Bot commented Jun 7, 2021

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b0015e7

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@cloudfour/patterns Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@tylersticka
tylersticka marked this pull request as ready for review June 7, 2021 21:41
@tylersticka
tylersticka requested a review from a team June 7, 2021 21:42

@Paul-Hebert Paul-Hebert left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM!

@tylersticka
tylersticka merged commit f24e17f into v-next Jun 7, 2021
@tylersticka
tylersticka deleted the feature/contained-cards-r2 branch June 7, 2021 22:33
@github-actions github-actions Bot mentioned this pull request Jun 7, 2021
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Implement card variations - Dev

2 participants