Skip to content

fix error CHANGELOG.md - #761

Merged
CarsonRoscoe merged 1 commit into
coinbase:mainfrom
Fallengirl:main
Jun 19, 2025
Merged

CarsonRoscoe merged 1 commit into
coinbase:mainfrom
Fallengirl:main

Conversation

@Fallengirl

Copy link
Copy Markdown
Contributor

hey team, fixed error in typescript/agentkit/CHANGELOG.md
exmaple - example

@Fallengirl
Fallengirl requested a review from murrlincoln as a code owner June 17, 2025 13:04
@cb-heimdall

cb-heimdall commented Jun 17, 2025 •

Copy link
Copy Markdown

✅ Heimdall Review Status

Requirement Status More Info
Reviews ✅ 1/1
Denominator calculation
Show calculation
1 if user is bot 0
1 if user is external 0
2 if repo is sensitive 0
From .codeflow.yml 1
Additional review requirements
Show calculation
Max 0
0
From CODEOWNERS 0
Global minimum 0
Max 1
1
1 if commit is unverified 0
Sum 1

@github-actions github-actions Bot added the documentation Improvements or additions to documentation label Jun 17, 2025
@CarsonRoscoe
CarsonRoscoe merged commit 294541d into coinbase:main Jun 19, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

documentation Improvements or additions to documentation

Development

Successfully merging this pull request may close these issues.

3 participants