Skip to content

fix elizaos-plugins reference - #768

Merged
CarsonRoscoe merged 1 commit into
coinbase:mainfrom
aso20455:main
Jul 24, 2025
Merged

CarsonRoscoe merged 1 commit into
coinbase:mainfrom
aso20455:main

Conversation

@aso20455

@aso20455 aso20455 commented Jul 2, 2025

Copy link
Copy Markdown
Contributor

@aso20455
aso20455 requested a review from murrlincoln as a code owner July 2, 2025 14:20
@cb-heimdall

cb-heimdall commented Jul 2, 2025 •

Copy link
Copy Markdown

✅ Heimdall Review Status

Requirement Status More Info
Reviews ✅ 1/1
Denominator calculation
Show calculation
1 if user is bot 0
1 if user is external 0
2 if repo is sensitive 0
From .codeflow.yml 1
Additional review requirements
Show calculation
Max 0
0
From CODEOWNERS 0
Global minimum 0
Max 1
1
1 if commit is unverified 0
Sum 1

@github-actions github-actions Bot added the documentation Improvements or additions to documentation label Jul 2, 2025
@CarsonRoscoe
CarsonRoscoe merged commit 0b7b24f into coinbase:main Jul 24, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

documentation Improvements or additions to documentation

Development

Successfully merging this pull request may close these issues.

3 participants