Skip to content

docs: fix typo in typescript/examples/langchain-zerodev-chatbot/README.md - #772

Merged
CarsonRoscoe merged 1 commit into
coinbase:mainfrom
lechpzn:1
Jul 24, 2025
Merged

CarsonRoscoe merged 1 commit into
coinbase:mainfrom
lechpzn:1

Conversation

@lechpzn

@lechpzn lechpzn commented Jul 9, 2025

Copy link
Copy Markdown
Contributor

Description

interexchangably ----> interchangeably

Tests

Checklist

A couple of things to include in your PR for completeness:

  • Added documentation to all relevant README.md files
  • Added a changelog entry

@lechpzn
lechpzn requested a review from murrlincoln as a code owner July 9, 2025 15:01
@cb-heimdall

cb-heimdall commented Jul 9, 2025 •

Copy link
Copy Markdown

✅ Heimdall Review Status

Requirement Status More Info
Reviews ✅ 1/1
Denominator calculation
Show calculation
1 if user is bot 0
1 if user is external 0
2 if repo is sensitive 0
From .codeflow.yml 1
Additional review requirements
Show calculation
Max 0
0
From CODEOWNERS 0
Global minimum 0
Max 1
1
1 if commit is unverified 0
Sum 1

@github-actions github-actions Bot added documentation Improvements or additions to documentation example New example agent labels Jul 9, 2025
@cb-heimdall

Copy link
Copy Markdown

Review Error for lavonnes @ 2025-07-15 11:10:39 UTC
User must have write permissions to review

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

documentation Improvements or additions to documentation example New example agent

Development

Successfully merging this pull request may close these issues.

4 participants