Skip to content

fix: changed changeset from minor to patch - #799

Merged
CarsonRoscoe merged 1 commit into
mainfrom
fix/patch-notes-typescript
Jul 25, 2025
Merged

CarsonRoscoe merged 1 commit into
mainfrom
fix/patch-notes-typescript

Conversation

@CarsonRoscoe

Copy link
Copy Markdown
Contributor

Fixing minor to patch in changesets

@cb-heimdall

cb-heimdall commented Jul 25, 2025 •

Copy link
Copy Markdown

✅ Heimdall Review Status

Requirement Status More Info
Reviews ✅ 1/1
Denominator calculation
Show calculation
1 if user is bot 0
1 if user is external 0
2 if repo is sensitive 0
From .codeflow.yml 1
Additional review requirements
Show calculation
Max 0
0
From CODEOWNERS 0
Global minimum 0
Max 1
1
1 if commit is unverified 0
Sum 1

@CarsonRoscoe
CarsonRoscoe marked this pull request as ready for review July 25, 2025 20:50
@CarsonRoscoe
CarsonRoscoe requested a review from 0xRAG July 25, 2025 20:50
@github-actions github-actions Bot added the documentation Improvements or additions to documentation label Jul 25, 2025
@CarsonRoscoe
CarsonRoscoe merged commit 6f17841 into main Jul 25, 2025
@CarsonRoscoe
CarsonRoscoe deleted the fix/patch-notes-typescript branch July 25, 2025 20:56
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

documentation Improvements or additions to documentation

Development

Successfully merging this pull request may close these issues.

3 participants