Skip to content

fix: bugs in next template configs - #809

Merged
CarsonRoscoe merged 1 commit into
coinbase:mainfrom
phdargen:next-template-fixes
Aug 1, 2025
Merged

CarsonRoscoe merged 1 commit into
coinbase:mainfrom
phdargen:next-template-fixes

Conversation

@phdargen

Copy link
Copy Markdown
Contributor

Fixes various build issues with the next templates due to renaming of WalletProviders and WalletProviderConfigs

Closes issue #790

@phdargen
phdargen requested a review from murrlincoln as a code owner July 31, 2025 18:10
@cb-heimdall

cb-heimdall commented Jul 31, 2025 •

Copy link
Copy Markdown

✅ Heimdall Review Status

Requirement Status More Info
Reviews ✅ 1/1
Denominator calculation
Show calculation
1 if user is bot 0
1 if user is external 0
2 if repo is sensitive 0
From .codeflow.yml 1
Additional review requirements
Show calculation
Max 0
0
From CODEOWNERS 0
Global minimum 0
Max 1
1
1 if commit is unverified 0
Sum 1

@github-actions github-actions Bot added documentation Improvements or additions to documentation typescript labels Jul 31, 2025

@CarsonRoscoe CarsonRoscoe left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Fantastic, thank you very much for catching these

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

documentation Improvements or additions to documentation typescript

Development

Successfully merging this pull request may close these issues.

3 participants