Skip to content

[release/11.0] Don't pass empty string as SNI to QUIC - #133155

Merged
svick merged 1 commit into
release/11.0from
backport/pr-133097-to-release/11.0
Sep 3, 2026
Merged

svick merged 1 commit into
release/11.0from
backport/pr-133097-to-release/11.0

Conversation

@github-actions

@github-actions github-actions Bot commented Sep 3, 2026 •

Copy link
Copy Markdown
Contributor

Backport of #133097 to release/11.0

/cc @rzikm

Customer Impact

  • Customer reported
  • Found internally

Customers would be unable to establish QUIC or HTTP/3 connections on Linux without providing the target host name (to verify against the server certificate).

Regression

  • Yes
  • No

Regression by updating MsQuic (optional dependency to light-up QUIC support)

Testing

Affected tests succeed on CI.

Risk

Low, change is small and well understood.

@azure-pipelines

Copy link
Copy Markdown
Azure Pipelines:
Successfully started running 3 pipeline(s).
13 pipeline(s) were filtered out due to trigger conditions.
There may be pipelines that require an authorized user to comment /azp run to run.

@dotnet-policy-service

Copy link
Copy Markdown
Contributor

Tagging subscribers to this area: @karelz, @dotnet/ncl
See info in area-owners.md if you want to be subscribed.

@rzikm rzikm self-assigned this Sep 3, 2026
@rzikm rzikm added Servicing-consider Issue for next servicing release review area-System.Net.Quic and removed area-System.Net labels Sep 3, 2026
@dotnet-policy-service

Copy link
Copy Markdown
Contributor

Tagging subscribers to this area: @karelz, @dotnet/ncl
See info in area-owners.md if you want to be subscribed.

@rzikm rzikm added Servicing-approved Approved for servicing release and removed Servicing-consider Issue for next servicing release review labels Sep 3, 2026
@rzikm

rzikm commented Sep 3, 2026

Copy link
Copy Markdown
Member

Approved by Jeff Schwartz via email

@rzikm

rzikm commented Sep 3, 2026

Copy link
Copy Markdown
Member

/ba-g Test failures are unrelated

@svick
svick merged commit 10f4939 into release/11.0 Sep 3, 2026
105 of 113 checks passed
@svick
svick deleted the backport/pr-133097-to-release/11.0 branch September 3, 2026 16:42
@dotnet-milestone-bot dotnet-milestone-bot Bot added this to the 11.0-rc2 milestone Sep 4, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

area-System.Net.Quic Servicing-approved Approved for servicing release

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants