Skip to content

fix(queue): remove incorrect !-assertion - #240

Merged
nbsp merged 2 commits into
nextfrom
nbsp/fix/queue-get
Jan 8, 2025
Merged

nbsp merged 2 commits into
nextfrom
nbsp/fix/queue-get

Conversation

@nbsp

@nbsp nbsp commented Jan 8, 2025

Copy link
Copy Markdown
Contributor

No description provided.

@nbsp
nbsp requested review from lukasIO and s-hamdananwar January 8, 2025 14:29
@changeset-bot

changeset-bot Bot commented Jan 8, 2025 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 499eae6

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@livekit/agents Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@s-hamdananwar s-hamdananwar left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

lgtm!

@nbsp
nbsp merged commit 4a66a82 into next Jan 8, 2025
@nbsp
nbsp deleted the nbsp/fix/queue-get branch January 8, 2025 17:25
@github-actions github-actions Bot mentioned this pull request Jan 14, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants