Skip to content

add eleven labs flash models - #246

Merged
nbsp merged 2 commits into
livekit:nextfrom
sam-trost:next
Jan 13, 2025
Merged

nbsp merged 2 commits into
livekit:nextfrom
sam-trost:next

Conversation

@sam-trost

Copy link
Copy Markdown
Contributor

Fixes #245

@changeset-bot

changeset-bot Bot commented Jan 13, 2025 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7fbe26c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@livekit/agents-plugin-elevenlabs Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@nbsp
nbsp merged commit 469a223 into livekit:next Jan 13, 2025
@sam-trost
sam-trost deleted the next branch January 13, 2025 15:40
@github-actions github-actions Bot mentioned this pull request Jan 14, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants