Skip to content

Added support for custom deepgram base url - #1088

Closed
milo157 wants to merge 10 commits into
livekit:mainfrom
CerebriumAI:feature/local-deepgram
Closed

milo157 wants to merge 10 commits into
livekit:mainfrom
CerebriumAI:feature/local-deepgram

Conversation

@milo157

@milo157 milo157 commented Nov 14, 2024

Copy link
Copy Markdown
Contributor

No description provided.

@changeset-bot

changeset-bot Bot commented Nov 14, 2024 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: e0fe5f4

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
livekit-plugins-deepgram Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@CLAassistant

CLAassistant commented Nov 14, 2024 •

Copy link
Copy Markdown

CLA assistant check
All committers have signed the CLA.

Comment thread livekit-plugins/livekit-plugins-deepgram/livekit/plugins/deepgram/stt.py Outdated

@theomonnom theomonnom left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks, lgtm! Can you run ruff format && ruff check --fix

@theomonnom

Copy link
Copy Markdown
Member

It seems like I don't have write access to this PR, can you click on the changesets comment to add one 🙏

@theomonnom theomonnom left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I don't think these lines are needed?

Comment thread livekit-plugins/livekit-plugins-deepgram/livekit/plugins/deepgram/stt.py Outdated
Comment thread livekit-plugins/livekit-plugins-deepgram/livekit/plugins/deepgram/stt.py Outdated
@theomonnom

Copy link
Copy Markdown
Member

Hey you removed the wrong lines 😅

@milo157

milo157 commented Nov 15, 2024

Copy link
Copy Markdown
Contributor Author

Okay everything should hopefully be resolved now

@milo157

milo157 commented Nov 18, 2024

Copy link
Copy Markdown
Contributor Author

@theomonnom @davidzhao are you able to re-review please

@milo157

milo157 commented Nov 20, 2024

Copy link
Copy Markdown
Contributor Author

@theomonnom @davidzhao can this be merged please?

@milo157

milo157 commented Nov 26, 2024

Copy link
Copy Markdown
Contributor Author

@theomonnom @davidzhao any update here please?

@theomonnom

Copy link
Copy Markdown
Member

@theomonnom @davidzhao any update here please?

Will take a look at it this evening

@theomonnom

Copy link
Copy Markdown
Member

replaced by #1137

@theomonnom theomonnom closed this Nov 26, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants