Skip to content

fix: update default model to chirp2 in google stt & update generate_reply method in gemini realtime - #1401

Merged
jayeshp19 merged 8 commits into
mainfrom
google-timeout
Jan 23, 2025
Merged

jayeshp19 merged 8 commits into
mainfrom
google-timeout

Conversation

@jayeshp19

Copy link
Copy Markdown
Contributor

This PR includes: fix for failing test cases for google stt and update create_response method in gemini realtime

@changeset-bot

changeset-bot Bot commented Jan 22, 2025 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 28e37c8

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
livekit-plugins-google Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@jayeshp19
jayeshp19 requested a review from a team January 22, 2025 12:31
@jayeshp19 jayeshp19 changed the title fix: remove timeout from grpc request in google stt/tts & update generate_reply method in gemini realtime fix: update default model to chirp2 in google stt & update generate_reply method in gemini realtime Jan 22, 2025
@jayeshp19
jayeshp19 merged commit 695f7b5 into main Jan 23, 2025
@jayeshp19
jayeshp19 deleted the google-timeout branch January 23, 2025 07:28
@brightsparc

Copy link
Copy Markdown
Contributor

I have an open PR which you might also want to review @jayeshp19 #1098

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants