fix: general integration readiness patches - #94
Merged
Conversation
🦋 Changeset detectedLatest commit: c9c874f The changes in this PR will be included in the next version bump. Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
|
The latest updates on your projects. Learn more about Vercel for Git ↗︎
|
stephancill
force-pushed
the
fix/integration-improvements
branch
from
November 24, 2023 15:55
359f1d0 to
13ef63e
Compare
davidfurlong
requested changes
Nov 24, 2023
stephancill
force-pushed
the
fix/integration-improvements
branch
from
November 25, 2023 17:15
13ef63e to
9e461bd
Compare
stephancill
force-pushed
the
fix/integration-improvements
branch
from
November 25, 2023 17:24
9e461bd to
6dcd2c3
Compare
2 tasks
stephancill
force-pushed
the
fix/integration-improvements
branch
from
November 30, 2023 10:12
e2f4131 to
56da768
Compare
stephancill
force-pushed
the
fix/integration-improvements
branch
from
December 6, 2023 17:45
15c5272 to
765a986
Compare
davidfurlong
approved these changes
Dec 6, 2023
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
spacingwithgapfor compatibility reasonscard.tsxbold,secondary,regularvariants to text componentFarcasterUserobject (opengraph response) or only a partial object with only the fid (metadata response)Includes/depends on #81
Outstanding integration issues:
video.tsxcomponent global style import causes issues with nextjs (only when not using app router)@next/imagecomponentFixes #98