Skip to content

feat(docs): add favicon - #2330

Merged
ghostdevv merged 1 commit into
npmx-dev:mainfrom
trueberryless:feat/docs-favicon
Apr 2, 2026
Merged

ghostdevv merged 1 commit into
npmx-dev:mainfrom
trueberryless:feat/docs-favicon

Conversation

@trueberryless

Copy link
Copy Markdown
Member

🧭 Context

Add favicon to https://docs.npmx.dev

📚 Description

Add new file in public/favicon.ico, which is default favicon placement.

image

@vercel

vercel Bot commented Mar 30, 2026 •

Copy link
Copy Markdown
Contributor

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Actions Updated (UTC)
docs.npmx.dev Ready Ready Preview, Comment Mar 30, 2026 0:09am
npmx.dev Ready Ready Preview, Comment Mar 30, 2026 0:09am
1 Skipped Deployment
Project Deployment Actions Updated (UTC)
npmx-lunaria Ignored Ignored Mar 30, 2026 0:09am

Request Review

@codecov

codecov Bot commented Mar 30, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ All tests successful. No failed tests found.

📢 Thoughts on this report? Let us know!

@ghostdevv
ghostdevv added this pull request to the merge queue Apr 2, 2026
Merged via the queue into npmx-dev:main with commit a28563b Apr 2, 2026
21 checks passed
@github-actions github-actions Bot mentioned this pull request Apr 2, 2026
@trueberryless
trueberryless deleted the feat/docs-favicon branch April 2, 2026 04:02
@serhalp serhalp modified the milestones: v0.8.0, v0.9.0 Apr 7, 2026
@serhalp serhalp added this to Releases Apr 7, 2026
@github-project-automation github-project-automation Bot moved this to Done in Releases Apr 7, 2026
@serhalp serhalp modified the milestones: v0.9.0, v0.8.0 Apr 7, 2026

This branch was successfully deployed

2 active deployments
Preview – npmx.dev — 07f22165 Deployed Mar 30, 2026 by vercel[bot]
Preview – docs.npmx.dev — 07f22165 Deployed Mar 30, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

Status: Released

Development

Successfully merging this pull request may close these issues.

4 participants