Skip to content

chore(deps): update devdependency knip to v6.35.1 - #3249

Merged
danielroe merged 1 commit into
mainfrom
renovate/all-linters
Sep 14, 2026
Merged

danielroe merged 1 commit into
mainfrom
renovate/all-linters

Conversation

@renovate

@renovate renovate Bot commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Confidence
knip (source) 6.34.0 → 6.35.1 age confidence

Release Notes

webpro-nl/knip (knip)

v6.35.1: Release 6.35.1

Compare Source

v6.35.0: Release 6.35.0

Compare Source


Configuration

📅 Schedule: (UTC)

  • Branch creation
    • "on Monday"
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@vercel

vercel Bot commented Sep 14, 2026 •

Copy link
Copy Markdown
Contributor

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Actions Updated
docs.npmx.dev Ready Ready Preview Sep 14, 2026 12:39am UTC
npmx.dev Ready Ready Preview Sep 14, 2026 12:39am UTC
1 Skipped Deployment
Project Deployment Actions Updated
npmx-lunaria Ignored Ignored Sep 14, 2026 12:39am UTC

Request Review

@renovate renovate Bot added the chore label Sep 14, 2026
@socket-security

Copy link
Copy Markdown

Review the following changes in direct dependencies. Learn more about Socket for GitHub.

Diff Package Supply Chain
Security
Vulnerability Quality Maintenance License
Updatedknip@​6.34.0 ⏵ 6.35.1991009596100

View full report

@codecov

codecov Bot commented Sep 14, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ All tests successful. No failed tests found.

📢 Thoughts on this report? Let us know!

@github-actions

Copy link
Copy Markdown

e18e dependency analysis

No dependency warnings found.

@danielroe
danielroe added this pull request to the merge queue Sep 14, 2026
Merged via the queue into main with commit c33b8c6 Sep 14, 2026
30 checks passed
@danielroe
danielroe deleted the renovate/all-linters branch September 14, 2026 08:06

This branch was successfully deployed

2 active deployments
Preview – npmx.dev — 02402d18 Deployed Sep 14, 2026 by vercel[bot]
Preview – docs.npmx.dev — 02402d18 Deployed Sep 14, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant