Skip to content

feat: OpenShift Virtualization Higher Density - #1630

Closed
fabiand wants to merge 14 commits into
openshift:masterfrom
fabiand:cnv-swap
Closed

fabiand wants to merge 14 commits into
openshift:masterfrom
fabiand:cnv-swap

Conversation

@fabiand

@fabiand fabiand commented May 27, 2024

Copy link
Copy Markdown

A feature describing CNV's path to higher density based on

  • phase 1: wasp
  • phase 2: kube swap

@fabiand

fabiand commented May 27, 2024

Copy link
Copy Markdown
Author

cc @enp0s3 @iholder101

@openshift-ci
openshift-ci Bot requested review from cybertron and runcom May 27, 2024 13:02

@enp0s3 enp0s3 left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@fabiand Thank you!

Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated

@iholder101 iholder101 left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thank you @fabiand!

Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated
@fabiand

fabiand commented Jun 3, 2024

Copy link
Copy Markdown
Author

@enp0s3 @iholder101 many thanks for your inputs! I've addressed them and resolved consversations.
Please reopen conversations if you disagree.

@openshift-ci

openshift-ci Bot commented Jun 3, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign tkashem for approval. For more information see the Kubernetes Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@fabiand

fabiand commented Jun 12, 2024

Copy link
Copy Markdown
Author

@fabiand: all tests passed!

Thank you so much.

@enp0s3 enp0s3 left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@fabiand Hi, did second round, have some few cooments/ideas

Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md Outdated
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md
Comment thread enhancements/kubelet/virtualization-higher-workload-density.md
fabiand added 14 commits June 26, 2024 10:40
A feature describing CNV's path to higher density based on

- phase 1: wasp
- phase 2: kube swap

Signed-off-by: Fabian Deutsch <fabiand@fedoraproject.org>
Signed-off-by: Fabian Deutsch <fabiand@fedoraproject.org>
Signed-off-by: Fabian Deutsch <fabiand@fedoraproject.org>
Signed-off-by: Fabian Deutsch <fabiand@fedoraproject.org>
Signed-off-by: Fabian Deutsch <fabiand@fedoraproject.org>
Signed-off-by: Fabian Deutsch <fabiand@fedoraproject.org>
Signed-off-by: Fabian Deutsch <fabiand@fedoraproject.org>
Signed-off-by: Fabian Deutsch <fabiand@fedoraproject.org>
Signed-off-by: Fabian Deutsch <fabiand@fedoraproject.org>
Signed-off-by: Fabian Deutsch <fabiand@fedoraproject.org>
Signed-off-by: Fabian Deutsch <fabiand@fedoraproject.org>
Signed-off-by: Fabian Deutsch <fabiand@fedoraproject.org>
Signed-off-by: Fabian Deutsch <fabiand@fedoraproject.org>
Signed-off-by: Fabian Deutsch <fabiand@fedoraproject.org>
@fabiand

fabiand commented Jun 26, 2024

Copy link
Copy Markdown
Author

All set.

@fabiand fabiand mentioned this pull request Jun 26, 2024
@openshift-ci

openshift-ci Bot commented Jun 26, 2024

Copy link
Copy Markdown
Contributor

@fabiand: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@iholder101

Copy link
Copy Markdown

Looks great to me!
/lgtm

@openshift-ci

openshift-ci Bot commented Jul 1, 2024

Copy link
Copy Markdown
Contributor

@iholder101: changing LGTM is restricted to collaborators

Details

In response to this:

Looks great to me!
/lgtm

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@openshift-bot

Copy link
Copy Markdown

Inactive enhancement proposals go stale after 28d of inactivity.

See https://github.com/openshift/enhancements#life-cycle for details.

Mark the proposal as fresh by commenting /remove-lifecycle stale.
Stale proposals rot after an additional 7d of inactivity and eventually close.
Exclude this proposal from closing by commenting /lifecycle frozen.

If this proposal is safe to close now please do so with /close.

/lifecycle stale

@openshift-ci openshift-ci Bot added the lifecycle/stale Denotes an issue or PR has remained open with no activity and has become stale. label Jul 30, 2024
@enp0s3

enp0s3 commented Jul 30, 2024

Copy link
Copy Markdown

/remove-lifecycle stale

@openshift-ci openshift-ci Bot removed the lifecycle/stale Denotes an issue or PR has remained open with no activity and has become stale. label Jul 30, 2024
@enp0s3

enp0s3 commented Jul 30, 2024

Copy link
Copy Markdown

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jul 30, 2024
@fabiand

fabiand commented Aug 13, 2024

Copy link
Copy Markdown
Author

@mrunalp can we get this merged or reviewed?

@openshift-bot

Copy link
Copy Markdown

Inactive enhancement proposals go stale after 28d of inactivity.

See https://github.com/openshift/enhancements#life-cycle for details.

Mark the proposal as fresh by commenting /remove-lifecycle stale.
Stale proposals rot after an additional 7d of inactivity and eventually close.
Exclude this proposal from closing by commenting /lifecycle frozen.

If this proposal is safe to close now please do so with /close.

/lifecycle stale

@openshift-ci openshift-ci Bot added the lifecycle/stale Denotes an issue or PR has remained open with no activity and has become stale. label Sep 10, 2024
@fabiand

fabiand commented Sep 10, 2024 via email

Copy link
Copy Markdown
Author

@openshift-ci openshift-ci Bot removed the lifecycle/stale Denotes an issue or PR has remained open with no activity and has become stale. label Sep 10, 2024
@iholder101

Copy link
Copy Markdown

/hold

We need to update the GA section w.r.t. our plan forward to restrict non-CNV users from accessing swap.
I will take ownership on the PR to do so. Will introduce a replacement PR very soon.

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Sep 10, 2024
@iholder101

Copy link
Copy Markdown

See this replacement PR: #1679.
This PR can be closed now.

/close

@openshift-ci

openshift-ci Bot commented Sep 17, 2024

Copy link
Copy Markdown
Contributor

@iholder101: You can't close an active issue/PR unless you authored it or you are a collaborator.

Details

In response to this:

See this replacement PR: #1679.
This PR can be closed now.

/close

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@openshift-bot

Copy link
Copy Markdown

Inactive enhancement proposals go stale after 28d of inactivity.

See https://github.com/openshift/enhancements#life-cycle for details.

Mark the proposal as fresh by commenting /remove-lifecycle stale.
Stale proposals rot after an additional 7d of inactivity and eventually close.
Exclude this proposal from closing by commenting /lifecycle frozen.

If this proposal is safe to close now please do so with /close.

/lifecycle stale

@openshift-ci openshift-ci Bot added the lifecycle/stale Denotes an issue or PR has remained open with no activity and has become stale. label Oct 16, 2024
@openshift-bot

Copy link
Copy Markdown

Stale enhancement proposals rot after 7d of inactivity.

See https://github.com/openshift/enhancements#life-cycle for details.

Mark the proposal as fresh by commenting /remove-lifecycle rotten.
Rotten proposals close after an additional 7d of inactivity.
Exclude this proposal from closing by commenting /lifecycle frozen.

If this proposal is safe to close now please do so with /close.

/lifecycle rotten
/remove-lifecycle stale

@openshift-ci openshift-ci Bot added lifecycle/rotten Denotes an issue or PR that has aged beyond stale and will be auto-closed. and removed lifecycle/stale Denotes an issue or PR has remained open with no activity and has become stale. labels Oct 23, 2024
@enp0s3

enp0s3 commented Oct 23, 2024

Copy link
Copy Markdown

@fabiand Hi, I think only you have the permission to close this PR

@openshift-bot

Copy link
Copy Markdown

Rotten enhancement proposals close after 7d of inactivity.

See https://github.com/openshift/enhancements#life-cycle for details.

Reopen the proposal by commenting /reopen.
Mark the proposal as fresh by commenting /remove-lifecycle rotten.
Exclude this proposal from closing again by commenting /lifecycle frozen.

/close

@openshift-ci openshift-ci Bot closed this Oct 31, 2024
@openshift-ci

openshift-ci Bot commented Oct 31, 2024

Copy link
Copy Markdown
Contributor

@openshift-bot: Closed this PR.

Details

In response to this:

Rotten enhancement proposals close after 7d of inactivity.

See https://github.com/openshift/enhancements#life-cycle for details.

Reopen the proposal by commenting /reopen.
Mark the proposal as fresh by commenting /remove-lifecycle rotten.
Exclude this proposal from closing again by commenting /lifecycle frozen.

/close

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. lgtm Indicates that a PR is ready to be merged. lifecycle/rotten Denotes an issue or PR that has aged beyond stale and will be auto-closed.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants