Skip to content

Move text input style overrides into sx prop - #1446

Merged
colebemis merged 4 commits into
mainfrom
colebemis-patch-2
Sep 21, 2021
Merged

colebemis merged 4 commits into
mainfrom
colebemis-patch-2

Conversation

@colebemis

@colebemis colebemis commented Sep 21, 2021 •

Copy link
Copy Markdown
Contributor

Problem

We recently removed system props from the TextInput component (#1414) but we still relied on TextInput system props in SelectPanel. This caused a visual regression in SelectPanel:

CleanShot 2021-09-21 at 11 30 29@2x

Solution

To fix this issue, I moved the TextInput style overrides into the sx prop instead of system props.

CleanShot 2021-09-21 at 11 38 50@2x

@changeset-bot

changeset-bot Bot commented Sep 21, 2021 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ec33297

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@primer/components Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

Comment thread src/SelectPanel/SelectPanel.tsx Outdated
@github-actions

github-actions Bot commented Sep 21, 2021 •

Copy link
Copy Markdown
Contributor

size-limit report 📦

Path Size
dist/browser.esm.js 51.71 KB (0%)
dist/browser.umd.js 52 KB (0%)

@colebemis
colebemis marked this pull request as ready for review September 21, 2021 18:39
@colebemis
colebemis requested review from a team and dgreif September 21, 2021 18:39

@jfuchs jfuchs left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Good catch, thanks. I wonder why this typechecked without error...

@colebemis
colebemis merged commit b4e5287 into main Sep 21, 2021
@colebemis
colebemis deleted the colebemis-patch-2 branch September 21, 2021 19:00
@primer-css primer-css mentioned this pull request Sep 21, 2021
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants