Skip to content

Update module github.com/pelletier/go-toml/v2 to v2.4.3 - #6596

Merged
rdimitrov merged 1 commit into
mainfrom
renovate/github.laiyagushi.com-pelletier-go-toml-v2-2.x
Sep 9, 2026
Merged

rdimitrov merged 1 commit into
mainfrom
renovate/github.laiyagushi.com-pelletier-go-toml-v2-2.x

Conversation

@renovate

@renovate renovate Bot commented Sep 9, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Confidence
github.com/pelletier/go-toml/v2 v2.3.1 → v2.4.3 age confidence

Release Notes

pelletier/go-toml (github.com/pelletier/go-toml/v2)

v2.4.3

Compare Source

What's Changed

What's new
Fixed bugs
Documentation
Other changes

New Contributors

Full Changelog: pelletier/go-toml@v2.4.2...v2.4.3

v2.4.2

Compare Source

What's Changed

Fixed bugs

Full Changelog: pelletier/go-toml@v2.4.1...v2.4.2

v2.4.1

Compare Source

What's Changed

Fixed bugs

Full Changelog: pelletier/go-toml@v2.4.0...v2.4.1

v2.4.0

Compare Source

What's Changed

What's new
Performance
Fixed bugs
Documentation

New Contributors

Full Changelog: pelletier/go-toml@v2.3.1...v2.4.0


Configuration

📅 Schedule: (UTC)

  • Branch creation
    • "every weekend"
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@renovate renovate Bot added the dependencies Pull requests that update a dependency file label Sep 9, 2026
@renovate
renovate Bot requested a review from JAORMX as a code owner September 9, 2026 16:55
@renovate renovate Bot added the dependencies Pull requests that update a dependency file label Sep 9, 2026
@github-actions github-actions Bot added the size/XS Extra small PR: < 100 lines changed label Sep 9, 2026
@rdimitrov

Copy link
Copy Markdown
Member

/retest

@codecov

codecov Bot commented Sep 9, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 78.79%. Comparing base (0972fb8) to head (13cb7ea).
⚠️ Report is 32 commits behind head on main.

Additional details and impacted files
@@            Coverage Diff             @@
##             main    #6596      +/-   ##
==========================================
- Coverage   78.84%   78.79%   -0.05%     
==========================================
  Files         778      778              
  Lines       77490    77490              
==========================================
- Hits        61095    61060      -35     
- Misses      16390    16425      +35     
  Partials        5        5              

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@rdimitrov
rdimitrov merged commit 9f7cf29 into main Sep 9, 2026
52 of 53 checks passed
@rdimitrov
rdimitrov deleted the renovate/github.laiyagushi.com-pelletier-go-toml-v2-2.x branch September 9, 2026 22:29
@github-actions github-actions Bot mentioned this pull request Sep 10, 2026
2 tasks
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file size/XS Extra small PR: < 100 lines changed

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant