Skip to content

fix: compute_hyperdisk_create_from_pool - #3808

Merged
subfuzion merged 2 commits into
mainfrom
fix-compute_hyperdisk_create_from_pool
Sep 3, 2024
Merged

subfuzion merged 2 commits into
mainfrom
fix-compute_hyperdisk_create_from_pool

Conversation

@gryczj

@gryczj gryczj commented Aug 27, 2024 •

Copy link
Copy Markdown
Contributor

Description

Fixes #

Note: Before submitting a pull request, please open an issue for discussion if you are not associated with Google.

Checklist

  • I have followed guidelines from CONTRIBUTING.MD and Samples Style Guide
  • Tests pass: npm test (see Testing)
  • Lint pass: npm run lint (see Style)
  • These samples need a new API enabled in testing projects to pass (let us know which ones)
  • These samples need a new/updated env vars in testing projects set to pass (let us know which ones)
  • This pull request is from a branch created directly off of GoogleCloudPlatform/nodejs-docs-samples. Not a fork.
  • This sample adds a new sample directory, and I updated the CODEOWNERS file with the codeowners for this sample
  • This sample adds a new sample directory, and I created GitHub Actions workflow for this sample
  • This sample adds a new Product API, and I updated the Blunderbuss issue/PR auto-assigner with the codeowners for this sample
  • Please merge this PR for me once it is approved

@gryczj
gryczj requested review from a team as code owners August 27, 2024 07:48
@gryczj gryczj self-assigned this Aug 27, 2024
@product-auto-label product-auto-label Bot added samples Issues that are directly related to samples. api: compute Issues related to the Compute Engine API. labels Aug 27, 2024
@gryczj gryczj added kokoro:force-run Add this label to force Kokoro to re-run the tests. kokoro:run Add this label to force Kokoro to re-run the tests. labels Aug 27, 2024
@gryczj
gryczj force-pushed the fix-compute_hyperdisk_create_from_pool branch from 5118ff5 to b14df12 Compare August 27, 2024 07:49
@kokoro-team kokoro-team removed kokoro:run Add this label to force Kokoro to re-run the tests. kokoro:force-run Add this label to force Kokoro to re-run the tests. labels Aug 27, 2024
@gryczj
gryczj force-pushed the fix-compute_hyperdisk_create_from_pool branch from 8763938 to 8a5b414 Compare August 27, 2024 08:41
@gryczj

gryczj commented Aug 27, 2024 •

Copy link
Copy Markdown
Contributor Author

@subfuzion My proposal is to merge tests for compute_hyperdisk_pool_create and compute_hyperdisk_create_from_pool.
First we test if pool was created correctly and then we use this pool to create a new hyperdisk. By doing it this way we reduce number of operations for storagePool(there is a possibility to create only 5 new pool per project per hour and 10 storage pools per day). When developing compute directory is it really easy to exceed that.
Additonaly I added logic to control disk and pool deletion and thank to this we could get rid of those timeouts in before() and after() methods.
What do you think about this solution?

@gryczj gryczj mentioned this pull request Aug 27, 2024
5 of 10 tasks
Comment thread compute/disks/createComputeHyperdiskPool.js

@subfuzion subfuzion left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This looks a lot cleaner, thanks! Looks like this will unblock #3807. 🤞

@subfuzion
subfuzion merged commit 0b900f5 into main Sep 3, 2024
@subfuzion
subfuzion deleted the fix-compute_hyperdisk_create_from_pool branch September 3, 2024 15:22
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

api: compute Issues related to the Compute Engine API. samples Issues that are directly related to samples.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants