Skip to content

feat: Added Unity UserFeedback Prefab - #14667

Merged
bitsandfoxes merged 4 commits into
masterfrom
feat/unity-feedback
Aug 27, 2025
Merged

bitsandfoxes merged 4 commits into
masterfrom
feat/unity-feedback

Conversation

@bitsandfoxes

Copy link
Copy Markdown
Contributor

No description provided.

@vercel

vercel Bot commented Aug 14, 2025 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Preview Comments Updated (UTC)
sentry-docs Ready Ready Preview Comment Aug 27, 2025 11:22am
1 Skipped Deployment
Project Deployment Preview Comments Updated (UTC)
develop-docs Ignored Ignored Preview Aug 27, 2025 11:22am

@codeowner-assignment
codeowner-assignment Bot requested a review from a team August 14, 2025 14:03
Comment thread docs/platforms/unity/user-feedback/index.mdx
Comment thread docs/platforms/unity/user-feedback/index.mdx Outdated
Comment thread docs/platforms/unity/user-feedback/index.mdx Outdated

@coolguyzone coolguyzone left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🤙

@codeowner-assignment
codeowner-assignment Bot requested a review from a team August 27, 2025 10:45
@codecov

codecov Bot commented Aug 27, 2025 •

Copy link
Copy Markdown

Bundle Report

Changes will increase total bundle size by 1.5kB (0.01%) ⬆️. This is within the configured threshold ✅

Detailed changes
Bundle name Size Change
sentry-docs-client-array-push 9.9MB -6 bytes (-0.0%) ⬇️
sentry-docs-server-cjs 12.45MB 1.51kB (0.01%) ⬆️

Affected Assets, Files, and Routes:

view changes for bundle: sentry-docs-client-array-push

Assets Changed:

Asset Name Size Change Total Size Change (%)
static/chunks/pages/_app-*.js -3 bytes 882.71kB -0.0%
static/chunks/8321-*.js -3 bytes 425.87kB -0.0%
server/middleware-*.js -5.55kB 1.0kB -84.74%
server/middleware-*.js 5.55kB 6.55kB 555.3% ⚠️
static/0aJZB220nxfy8jy3isxh0/_buildManifest.js (New) 684 bytes 684 bytes 100.0% 🚀
static/0aJZB220nxfy8jy3isxh0/_ssgManifest.js (New) 77 bytes 77 bytes 100.0% 🚀
static/Cs4c_58xP67t4t6UHoZMY/_buildManifest.js (Deleted) -684 bytes 0 bytes -100.0% 🗑️
static/Cs4c_58xP67t4t6UHoZMY/_ssgManifest.js (Deleted) -77 bytes 0 bytes -100.0% 🗑️
view changes for bundle: sentry-docs-server-cjs

Assets Changed:

Asset Name Size Change Total Size Change (%)
1729.js -3 bytes 1.78MB -0.0%
../instrumentation.js -3 bytes 1.1MB -0.0%
9523.js -3 bytes 1.08MB -0.0%
../app/[[...path]]/page.js.nft.json 505 bytes 755.03kB 0.07%
../app/platform-redirect/page.js.nft.json 505 bytes 754.95kB 0.07%
../app/sitemap.xml/route.js.nft.json 505 bytes 752.42kB 0.07%

@bitsandfoxes
bitsandfoxes merged commit 4154599 into master Aug 27, 2025
13 checks passed
@bitsandfoxes
bitsandfoxes deleted the feat/unity-feedback branch August 27, 2025 11:32
codyde added a commit that referenced this pull request Aug 29, 2025
* 'master' of https://github.com/getsentry/sentry-docs: (72 commits)
  docs(replay): Bring replay nav item to all apple platforms (#14710)
  User Feedback Basics Page (#14670)
  Update AM3 and legacy billing for logs GA, DO NOT MERGE BEFORE LOGS GA (#14589)
  Add LLMSteering component for hidden documentation instructions (#14744)
  New data retention policy page (#14661)
  Adding error prediction to Prevent AI docs (#14641)
  feat(logs): Remove beta note from logs docs (#14686)
  First pass at logs quota doc (#14671)
  docs(aws): Remove manual setup steps for ESM (#14738)
  Android Logs docs without Errors / Breadcrumbs distractions (#14733)
  chore(Unity): Dropped 2019 relevant section from Known Limitations (#14735)
  feat: Added Unity `UserFeedback` Prefab (#14667)
  chore(Unity): Added Mono & Line Numbers to Known Limitations (#14736)
  docs(js): Fix Next.js client tracing config (#14734)
  chore: Updated known limitation in Unity with Cysharp (#14666)
  feat(Unity ): Added `APK+OBB` initialization to troubleshooting (#14668)
  Codecov copy & screenshot edits (#14659)
  fix(cloudflare): Remove unused instrument.mjs from hono on cloudflare (#14723)
  update advanced scrubbing doc - adding logs section (#14696)
  fixing some errors from nextjs-source-map-updates branch (#14467)
  ...
@github-actions github-actions Bot locked and limited conversation to collaborators Sep 12, 2025

This branch was successfully deployed

1 active deployment
Preview – sentry-docs — 529f6380 Deployed Aug 27, 2025 by vercel[bot]
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants